Skip to content
This repository was archived by the owner on Mar 21, 2025. It is now read-only.

enhance: Support passing trace ctx and displaying traceID in case of request err - #865

Open
tinswzy wants to merge 1 commit into
milvus-io:masterfrom
tinswzy:support_client_passing_trace
Open

enhance: Support passing trace ctx and displaying traceID in case of request err#865
tinswzy wants to merge 1 commit into
milvus-io:masterfrom
tinswzy:support_client_passing_trace

Conversation

@tinswzy

@tinswzy tinswzy commented Dec 20, 2024

Copy link
Copy Markdown

issue: #864
Support passing trace ctx and displaying traceID in case of request err

@sre-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: tinswzy
To complete the pull request process, please assign yah01 after the PR has been reviewed.
You can assign the PR to them by writing /assign @yah01 in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sre-ci-robot sre-ci-robot added the size/XL Denotes a PR that changes 500-999 lines. label Dec 20, 2024
@tinswzy
tinswzy force-pushed the support_client_passing_trace branch from c674d82 to 8035b80 Compare December 20, 2024 08:16
@sre-ci-robot sre-ci-robot added size/XXL Denotes a PR that changes 1000+ lines. and removed size/XL Denotes a PR that changes 500-999 lines. labels Dec 20, 2024
@codecov

codecov Bot commented Dec 20, 2024

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 0.00%. Comparing base (5a960e9) to head (8035b80).

Additional details and impacted files

Impacted file tree graph

@@      Coverage Diff      @@
##   master   #865   +/-   ##
=============================
=============================

@tinswzy
tinswzy force-pushed the support_client_passing_trace branch from 8035b80 to d5e9964 Compare December 20, 2024 09:07
@sre-ci-robot sre-ci-robot added size/XL Denotes a PR that changes 500-999 lines. and removed size/XXL Denotes a PR that changes 1000+ lines. labels Dec 20, 2024
@tinswzy
tinswzy force-pushed the support_client_passing_trace branch 2 times, most recently from 8d7a586 to 657d9bd Compare December 27, 2024 08:38
@tinswzy
tinswzy force-pushed the support_client_passing_trace branch 4 times, most recently from a6abbf1 to b6ed3cd Compare December 27, 2024 15:43
@tinswzy
tinswzy force-pushed the support_client_passing_trace branch from b6ed3cd to b04192c Compare January 6, 2025 01:47
@sre-ci-robot sre-ci-robot added size/L Denotes a PR that changes 100-499 lines. and removed size/XL Denotes a PR that changes 500-999 lines. labels Jan 6, 2025
Signed-off-by: tinswzy <zhenyuan.wei@zilliz.com>
@tinswzy
tinswzy force-pushed the support_client_passing_trace branch from b04192c to 6c0877a Compare January 11, 2025 02:28
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

size/L Denotes a PR that changes 100-499 lines.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants