fix: add codecov sonar cloud - #1253
Conversation
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #1253 +/- ##
=======================================
Coverage ? 62.22%
=======================================
Files ? 50
Lines ? 7208
Branches ? 0
=======================================
Hits ? 4485
Misses ? 2302
Partials ? 421 ☔ View full report in Codecov by Sentry. 🚀 New features to boost your workflow:
|
|
@flacatus: The following test has Failed, say /retest to rerun failed tests.
Inspecting Test ArtifactsTo inspect your test artifacts, follow these steps:
mkdir -p oras-artifacts
cd oras-artifacts
oras pull quay.io/konflux-test-storage/konflux-team/integration-service:konflux-e2e-5bz8lTest results analysis🚨 Failed to provision a cluster, see the log for more details: Click to view logsINFO: Log in to your Red Hat account... INFO: Configure AWS Credentials... INFO: Logged in as 'konflux-ci-418295695583' on 'https://api.openshift.com' INFO: Create ROSA with HCP cluster... time=2025-08-04T11:22:23Z level=info msg=Ignored check for policy key 'sts_hcp_ec2_registry_permission_policy' (zero egress feature toggle is not enabled) INFO: Creating cluster 'kx-fb51017f0c' INFO: To view a list of clusters and their status, run 'rosa list clusters' INFO: Cluster 'kx-fb51017f0c' has been created. INFO: Once the cluster is installed you will need to add an Identity Provider before you can login into the cluster. See 'rosa create idp --help' for more information.
Nodes:
INFO: Preparing to create operator roles. |
Maintainers will complete the following section