This repository was archived by the owner on Nov 28, 2025. It is now read-only.
fix(deps): update module github.com/getkin/kin-openapi to v0.133.0 - autoclosed - #859
Closed
red-hat-konflux[bot] wants to merge 1 commit into
Closed
Conversation
Contributor
Author
ℹ Artifact update noticeFile name: go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
|
Member
|
/retest |
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
2 times, most recently
from
March 16, 2025 09:08
73dd69e to
314e48a
Compare
Member
|
/retest |
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
from
March 20, 2025 12:54
314e48a to
78a1480
Compare
Member
|
/retest |
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
from
March 30, 2025 10:17
78a1480 to
89ac54e
Compare
Member
|
/retest |
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
from
April 6, 2025 06:34
89ac54e to
1e804f4
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
from
June 15, 2025 06:52
1e804f4 to
7e29fee
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
from
June 22, 2025 09:41
7e29fee to
8b2e1db
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
2 times, most recently
from
July 20, 2025 07:01
46c8b0f to
a9c08fe
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
from
July 27, 2025 06:19
a9c08fe to
2a3365a
Compare
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
from
August 24, 2025 08:42
2a3365a to
509e0c1
Compare
Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
red-hat-konflux
Bot
force-pushed
the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
from
August 31, 2025 08:53
509e0c1 to
aa61c8c
Compare
red-hat-konflux
Bot
deleted the
konflux/mintmaker/main/github.com-getkin-kin-openapi-0.x
branch
October 7, 2025 16:16
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to subscribe to this conversation on GitHub.
Already have an account?
Sign in.
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
v0.128.0->v0.133.0Warning
Some dependencies could not be looked up. Check the warning logs for more information.
Release Notes
getkin/kin-openapi (github.com/getkin/kin-openapi)
v0.133.0Compare Source
What's Changed
New Contributors
Full Changelog: getkin/kin-openapi@v0.132.0...v0.133.0
v0.132.0Compare Source
What's Changed
AuthenticatorFuncby @jamietanna in #1064New Contributors
Full Changelog: getkin/kin-openapi@v0.131.0...v0.132.0
v0.131.0Compare Source
What's Changed
Full Changelog: getkin/kin-openapi@v0.130.0...v0.131.0
v0.130.0Compare Source
What's Changed
New Contributors
Full Changelog: getkin/kin-openapi@v0.129.0...v0.130.0
v0.129.0Compare Source
What's Changed
x-www-form-urlencodedwith arbitrary nested allOf by @mikhalytch in #1046New Contributors
Full Changelog: getkin/kin-openapi@v0.128.0...v0.129.0
Configuration
📅 Schedule: Branch creation - "on Monday after 3am and before 10am" (UTC), Automerge - At any time (no schedule defined).
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
To execute skipped test pipelines write comment
/ok-to-test.This PR has been generated by MintMaker (powered by Renovate Bot).